Skip to content

Bump the bundler group with 2 updates - #196

Merged
dduugg merged 1 commit into
mainfrom
dependabot/bundler/bundler-01c42781e3
Oct 1, 2026
Merged

dduugg merged 1 commit into
mainfrom
dependabot/bundler/bundler-01c42781e3

Conversation

@dependabot

@dependabot dependabot Bot commented on behalf of github Oct 1, 2026

Copy link
Copy Markdown
Contributor

Bumps the bundler group with 2 updates: rubocop-gusto and spoom.

Updates rubocop-gusto from 11.8.1 to 11.9.0

Release notes

Sourced from rubocop-gusto's releases.

v11.9.0

11.9.0 (2026-09-24)

Features

  • add Gusto/HardcodedAbsentRecordId cop (#166) (f38282a)
  • add Gusto/HardcodedId cop (#167) (4cf3fff)
  • enforce with_fixed_indentation for Layout/ArgumentAlignment (#165) (2e4d8e7)
  • flag have_received and allow and_raise in Sidekiq/PerformAsyncStub (#174) (f6e2289)
Changelog

Sourced from rubocop-gusto's changelog.

11.9.0 (2026-09-24)

Features

  • add Gusto/HardcodedAbsentRecordId cop (#166) (f38282a)
  • add Gusto/HardcodedId cop (#167) (4cf3fff)
  • enforce with_fixed_indentation for Layout/ArgumentAlignment (#165) (2e4d8e7)
  • flag have_received and allow and_raise in Sidekiq/PerformAsyncStub (#174) (f6e2289)
Commits
  • caab29c chore(main): release 11.9.0 (#178)
  • f38282a feat: add Gusto/HardcodedAbsentRecordId cop (#166)
  • 2e4d8e7 feat: enforce with_fixed_indentation for Layout/ArgumentAlignment (#165)
  • f6e2289 feat: flag have_received and allow and_raise in Sidekiq/PerformAsyncStub (#174)
  • 4cf3fff feat: add Gusto/HardcodedId cop (#167)
  • 9bb2bd1 chore: Delete .github/CODEOWNERS (#177)
  • See full diff in compare view

Updates spoom from 1.8.7 to 1.8.8

Release notes

Sourced from spoom's releases.

v1.8.8

What's Changed

🛠 Other Changes

Full Changelog: Shopify/spoom@v1.8.7...v1.8.8

Commits
  • eb3c7db Bump version to v1.8.8
  • fbd9fc3 Merge pull request #1025 from Shopify/at/abstract-without-runtime-fix
  • 230854b Preserve skipped abstract signatures without runtime
  • 0c099b6 Merge pull request #1023 from Shopify/dependabot/github_actions/ruby/setup-ru...
  • e438edb Bump ruby/setup-ruby from 1.321.0 to 1.324.0
  • 42d4a44 Merge pull request #1022 from Shopify/dependabot/bundler/minor-and-patch-5634...
  • 12e941f Bump sorbet-static-and-runtime in the minor-and-patch group
  • dd20b7b Merge pull request #1021 from Shopify/dependabot/bundler/minor-and-patch-3cfe...
  • 2dd4f80 Make LSP diagnostics test order independent
  • 87854b3 Bump the minor-and-patch group with 2 updates
  • Additional commits viewable in compare view

Dependabot will resolve any conflicts with this PR as long as you don't alter it yourself. You can also trigger a rebase manually by commenting @dependabot rebase.


Dependabot commands and options

You can trigger Dependabot actions by commenting on this PR:

  • @dependabot rebase will rebase this PR
  • @dependabot recreate will recreate this PR, overwriting any edits that have been made to it
  • @dependabot show <dependency name> ignore conditions will show all of the ignore conditions of the specified dependency
  • @dependabot ignore <dependency name> major version will close this group update PR and stop Dependabot creating any more for the specific dependency's major version (unless you unignore this specific dependency's major version or upgrade to it yourself)
  • @dependabot ignore <dependency name> minor version will close this group update PR and stop Dependabot creating any more for the specific dependency's minor version (unless you unignore this specific dependency's minor version or upgrade to it yourself)
  • @dependabot ignore <dependency name> will close this group update PR and stop Dependabot creating any more for the specific dependency (unless you unignore this specific dependency or upgrade to it yourself)
  • @dependabot unignore <dependency name> will remove all of the ignore conditions of the specified dependency
  • @dependabot unignore <dependency name> <ignore condition> will remove the ignore condition of the specified dependency and ignore conditions

Bumps the bundler group with 2 updates: [rubocop-gusto](https://github.com/gusto/rubocop-gusto) and [spoom](https://github.com/Shopify/spoom).


Updates `rubocop-gusto` from 11.8.1 to 11.9.0
- [Release notes](https://github.com/gusto/rubocop-gusto/releases)
- [Changelog](https://github.com/Gusto/rubocop-gusto/blob/main/CHANGELOG.md)
- [Commits](Gusto/rubocop-gusto@v11.8.1...v11.9.0)

Updates `spoom` from 1.8.7 to 1.8.8
- [Release notes](https://github.com/Shopify/spoom/releases)
- [Commits](Shopify/spoom@v1.8.7...v1.8.8)

---
updated-dependencies:
- dependency-name: rubocop-gusto
  dependency-version: 11.9.0
  dependency-type: direct:development
  update-type: version-update:semver-minor
  dependency-group: bundler
- dependency-name: spoom
  dependency-version: 1.8.8
  dependency-type: direct:development
  update-type: version-update:semver-patch
  dependency-group: bundler
...

Signed-off-by: dependabot[bot] <support@github.com>
@dependabot dependabot Bot added dependencies Pull requests that update a dependency file ruby Pull requests that update ruby code labels Oct 1, 2026
@dependabot
dependabot Bot requested a review from a team as a code owner October 1, 2026 08:58
@dependabot dependabot Bot added dependencies Pull requests that update a dependency file ruby Pull requests that update ruby code labels Oct 1, 2026
@dduugg
dduugg merged commit c017bdb into main Oct 1, 2026
11 checks passed
@dduugg
dduugg deleted the dependabot/bundler/bundler-01c42781e3 branch October 1, 2026 16:28
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

dependencies Pull requests that update a dependency file ruby Pull requests that update ruby code

Projects

Status: Done

Development

Successfully merging this pull request may close these issues.

1 participant